Skip to content

next: PostGraphile v5.2 - #3127

Open
benjie wants to merge 357 commits into
mainfrom
next
Open

benjie wants to merge 357 commits into
mainfrom
next

Conversation

@benjie

@benjie benjie commented Aug 5, 2026

Copy link
Copy Markdown
Member

These are the semver minor changes merged into the next branch that should make up the PostGraphile v5.2.0 release. Holding off of merging these because we're expecting a few.

@changeset-bot

changeset-bot Bot commented Aug 5, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0e9c084

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 25 packages
Name Type
grafast Minor
graphile-utils Minor
postgraphile Minor
@dataplan/pg Minor
graphile-build-pg Minor
@dataplan/json Patch
@grafserv/persisted Patch
@graphile/lru Patch
@graphile/simplify-inflection Patch
eslint-plugin-graphile-export Patch
grafserv Patch
graphile Patch
graphile-build Minor
graphile-config Patch
graphile-export Patch
graphql-codegen-grafast Patch
jest-serializer-graphql-schema Patch
jest-serializer-simple Patch
pg-introspection Patch
pg-sql2 Patch
pgl Patch
ruru Patch
ruru-components Patch
ruru-types Patch
tamedevil Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@benjie benjie added this to the 5.2 milestone Aug 5, 2026
benjaie and others added 30 commits September 19, 2026 18:11
… so you get your chosen adaptor showing up in the types automatically.
…anning but does not execute the operation. Useful for pre-warming the operation plan cache on server start.

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants